Skip to content

Added the new columns as requested - #155

Merged
LeandroHamaguchi merged 5 commits into
mainfrom
new-fields
May 20, 2026
Merged

LeandroHamaguchi merged 5 commits into
mainfrom
new-fields

Conversation

@LeandroHamaguchi

Copy link
Copy Markdown
Member

No description provided.

… references and enhancing role types to include training. Revise UI elements and instructions for clarity, ensuring consistent terminology across components.
@vercel

vercel Bot commented May 10, 2026 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Actions Updated (UTC)
trccmwar Ready Ready Preview, Comment May 20, 2026 7:58pm
1 Skipped Deployment
Project Deployment Actions Updated (UTC)
trcc Ignored Ignored May 20, 2026 7:58pm

@LeandroHamaguchi
LeandroHamaguchi merged commit b15b470 into main May 20, 2026
5 checks passed

This branch was successfully deployed

1 active deployment
Preview – trccmwar — 0e445e32 Deployed May 20, 2026 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants