Repository navigation
fix filesystem namespaceStore creation - #3079
Conversation
|
@Soundarya189r: This pull request references [Jira Issue DFBUGS-6755](https://redhat.atlassian.net/browse/DFBUGS-6755), which is valid. The bug has been moved to the POST state. 2 validation(s) were run on this bug
No GitHub users were found matching the public email listed for the QA contact in Jira (ukurundw@redhat.com), skipping review request. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
Warning Review limit reachedNext included review available in 53 minutes. View limit detailsLimit details: You’ve used the included review currently available. You've used all free OSS reviews for now. Wait for the free limit to reset to keep reviewing this public repository. Review configuration: ⚙️ Run configurationConfiguration used: defaults Review profile: CHILL Plan: Advanced Run ID: 📒 Files selected for processing (1)
Thanks for using CodeRabbit! It's free for OSS, and your support helps us grow. If you like it, consider giving us a shout-out. Comment |
|
/test odf-console-e2e-aws |
|
@Soundarya189r: This pull request references [Jira Issue DFBUGS-6755](https://redhat.atlassian.net/browse/DFBUGS-6755), which is valid. 2 validation(s) were run on this bug
No GitHub users were found matching the public email listed for the QA contact in Jira (ukurundw@redhat.com), skipping review request. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
3097b07 to
2e9d062
Compare
|
/retitle fix filesystem namespaceStore creation |
|
@Soundarya189r: No Jira issue is referenced in the title of this pull request. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
SanjalKatiyar
left a comment
There was a problem hiding this comment.
Is this confirmed by NooBaa team ??
yeah, please check |
2e9d062 to
1cc1852
Compare
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: SanjalKatiyar, Soundarya189r The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/cherry-pick release-5.0 |
|
@SanjalKatiyar: once the present PR merges, I will cherry-pick it on top of DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
/test odf-console-e2e-aws |
1 similar comment
|
/test odf-console-e2e-aws |
744e1df
into
red-hat-storage:master
|
@SanjalKatiyar: new pull request created: #3119 DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
/jira backport release-4.22 |
|
@Soundarya189r: The following backport issues have been created: Queuing cherrypicks to the requested branches to be created after this PR merges: DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
@openshift-ci-robot: new pull request created: #3131 DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
/cherrypick release-4.22-compatibility |
|
@Soundarya189r: new pull request created: #3132 DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |



Description
Updated the filter to allow only PVCs that are:
Bound
ReadWriteMany
Filesystem volume mode
The reason for this change is to make sure users can select only Filesystem PVCs for creating a Filesystem NamespaceStore.
Change Type
Please select all applicable options:
Component / Area Impacted
Please select all applicable options:
Screenshots / Recordings
Screenshots
Testing
Please select the type of tests included:
Additional Notes