chore(deps): update registry.access.redhat.com/ubi8/ubi-minimal:latest docker digest to 4e4bb2f - #3270
chore(deps): update registry.access.redhat.com/ubi8/ubi-minimal:latest docker digest to 4e4bb2f#3270red-hat-konflux[bot] wants to merge 1 commit into
Conversation
|
There are test jobs defined for this repository which are not configured to run automatically. Comment |
|
Important Review skippedAuto reviews are disabled on this repository. Please check the settings in the CodeRabbit UI or the ⚙️ Run configurationConfiguration used: Repository YAML (base), Central YAML (inherited) Review profile: CHILL Plan: Enterprise Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
✨ Finishing Touches🧪 Generate unit tests (beta)
Comment |
|
[APPROVALNOTIFIER] This PR is APPROVED Approval requirements bypassed by manually added approval. This pull-request has been approved by: red-hat-konflux[bot] The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
fdcfed8 to
46ed184
Compare
|
New changes are detected. LGTM label has been removed. |
46ed184 to
4d9c47d
Compare
4d9c47d to
abadea7
Compare
abadea7 to
e1240e6
Compare
e1240e6 to
65719a5
Compare
65719a5 to
a270862
Compare
a270862 to
ba76eff
Compare
ba76eff to
88c2e6f
Compare
…t docker digest to 4e4bb2f Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
88c2e6f to
cc668e5
Compare
This PR contains the following updates:
5d76d0c→4e4bb2fWarning
Some dependencies could not be looked up. Check the Dependency Dashboard for more information.
Configuration
📅 Schedule: (in timezone UTC)
* 2-4 * * 1-5)🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
To execute skipped test pipelines write comment
/ok-to-test.Documentation
Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.