Skip to content

Add members 2026-09-05 - #6613

Open
jasonbraganza wants to merge 1 commit into
kubernetes:mainfrom
jasonbraganza:add-members-2026-09-05
Open

Add members 2026-09-05#6613
jasonbraganza wants to merge 1 commit into
kubernetes:mainfrom
jasonbraganza:add-members-2026-09-05

Conversation

@jasonbraganza

Copy link
Copy Markdown
Member

@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: jasonbraganza

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Sep 5, 2026
@kubernetes-prow kubernetes-prow Bot added approved Indicates a PR has been approved by an approver from all required OWNERS files. area/github-management Issues or PRs related to GitHub Management subproject size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Sep 5, 2026
@jasonbraganza

Copy link
Copy Markdown
Member Author

/hold

for review and lgtm by github admins.

@kubernetes-prow kubernetes-prow Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Sep 5, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/github-management Issues or PRs related to GitHub Management subproject cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

REQUEST: New membership for @MichalZylinski

1 participant