Skip to content

Add v1.38 Release Team permissions and clean up older releases org - #6607

Open
sreeram-venkitesh wants to merge 3 commits into
kubernetes:mainfrom
sreeram-venkitesh:update-v1.38-release-team
Open

Add v1.38 Release Team permissions and clean up older releases org#6607
sreeram-venkitesh wants to merge 3 commits into
kubernetes:mainfrom
sreeram-venkitesh:update-v1.38-release-team

Conversation

@sreeram-venkitesh

Copy link
Copy Markdown
Member
  • Add v1.38 Release Team Lead, Lead Shadows and Subteam leads to various teams.
  • Cleanup members from previous releases.

Ref: kubernetes/sig-release#3087

@kubernetes-prow kubernetes-prow Bot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Sep 3, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: sreeram-venkitesh
Once this PR has been reviewed and has the lgtm label, please assign puerco for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added area/github-management Issues or PRs related to GitHub Management subproject sig/release Categorizes an issue or PR as relevant to SIG Release. size/M Denotes a PR that changes 30-99 lines, ignoring generated files. labels Sep 3, 2026
@sreeram-venkitesh

Copy link
Copy Markdown
Member Author

/cc katcosgrove fsmunoz

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

area/github-management Issues or PRs related to GitHub Management subproject cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. sig/release Categorizes an issue or PR as relevant to SIG Release. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant